Skip to content

chore(deps): bump redis from 1.2.1 to 1.6.0 - #240

Closed
dependabot[bot] wants to merge 1 commit into
masterfrom
dependabot/cargo/redis-1.6.0
Closed

dependabot[bot] wants to merge 1 commit into
masterfrom
dependabot/cargo/redis-1.6.0

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Aug 24, 2026 •

Copy link
Copy Markdown
Contributor

Bumps redis from 1.2.1 to 1.6.0.

Release notes

Sourced from redis's releases.

redis-1.6.0

Changes & Bug fixes

CI & operational improvements

New Contributors

Full Changelog: redis-rs/redis-rs@redis-1.5.0...redis-1.5.1

redis-1.5.0

Changes & Bug fixes

CI & operational improvements

redis-1.4.1

Changes & Bug fixes

  • Remove unnecessary sleep from cluster readonly error handling (#2223 by @​nihohit)

CI & operational improvements

  • ci: Allow semicolon_in_expressions_from_macros to make nightly pass (#2218 by @​somechris)
  • tests/acl: Add the missing requirements for token based authentication (#2214 by @​somechris)
  • Makefile: Switch to --locked for module tests on RESP3 (#2222 by @​somechris)

Full Changelog: redis-rs/redis-rs@redis-1.4.0...redis-1.4.1

redis-1.4.0

Changes & Bug fixes

... (truncated)

Commits

@dependabot dependabot Bot added dependencies Pull requests that update a dependency file rust Pull requests that update rust code labels Aug 24, 2026
@verseghy-prow
verseghy-prow Bot requested a review from TwoDCube August 24, 2026 04:30
@verseghy-prow

verseghy-prow Bot commented Aug 24, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: dependabot[bot]
Once this PR has been reviewed and has the lgtm label, please assign smrtrfszm for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@TwoDCube

Copy link
Copy Markdown
Member

/un-cc
/cc @smrtrfszm

@verseghy-prow
verseghy-prow Bot requested a review from smrtrfszm August 24, 2026 10:41
@TwoDCube

Copy link
Copy Markdown
Member

/un-cc @TwoDCube

@TwoDCube

Copy link
Copy Markdown
Member

/uncc

@verseghy-prow
verseghy-prow Bot removed the request for review from TwoDCube August 24, 2026 10:42
Bumps [redis](https://github.com/redis-rs/redis-rs) from 1.2.1 to 1.6.0.
- [Release notes](https://github.com/redis-rs/redis-rs/releases)
- [Commits](redis-rs/redis-rs@redis-1.2.1...redis-1.6.0)

---
updated-dependencies:
- dependency-name: redis
  dependency-version: 1.6.0
  dependency-type: direct:production
  update-type: version-update:semver-minor
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot
dependabot Bot force-pushed the dependabot/cargo/redis-1.6.0 branch from 5cfcfdd to b546d72 Compare August 28, 2026 15:40
@dependabot @github

dependabot Bot commented on behalf of github Sep 14, 2026

Copy link
Copy Markdown
Contributor Author

Superseded by #245.

@dependabot dependabot Bot closed this Sep 14, 2026
@dependabot
dependabot Bot deleted the dependabot/cargo/redis-1.6.0 branch September 14, 2026 04:31
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dependencies Pull requests that update a dependency file rust Pull requests that update rust code

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant