Skip to content

chore(deps): bump redis from 1.2.1 to 1.5.0 - #238

Closed
dependabot[bot] wants to merge 1 commit into
masterfrom
dependabot/cargo/redis-1.5.0
Closed

dependabot[bot] wants to merge 1 commit into
masterfrom
dependabot/cargo/redis-1.5.0

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Aug 3, 2026

Copy link
Copy Markdown
Contributor

Bumps redis from 1.2.1 to 1.5.0.

Release notes

Sourced from redis's releases.

redis-1.5.0

Changes & Bug fixes

CI & operational improvements

redis-1.4.1

Changes & Bug fixes

  • Remove unnecessary sleep from cluster readonly error handling (#2223 by @​nihohit)

CI & operational improvements

  • ci: Allow semicolon_in_expressions_from_macros to make nightly pass (#2218 by @​somechris)
  • tests/acl: Add the missing requirements for token based authentication (#2214 by @​somechris)
  • Makefile: Switch to --locked for module tests on RESP3 (#2222 by @​somechris)

Full Changelog: redis-rs/redis-rs@redis-1.4.0...redis-1.4.1

redis-1.4.0

Changes & Bug fixes

CI & operational improvements

New Contributors

... (truncated)

Commits
  • 65a344a Prepare new version (#2267)
  • b9e2916 Add support for the new FPHA storage hint (#2114)
  • c8297d1 Fix nightly linter (#2258)
  • 30285c8 docs: Add the need for non_exhaustive for enums and structs (#2247)
  • 6ae8530 Async cluster connection: Reconnect after receiving push disconnection. (#2242)
  • e4e2dcf fix changelog (#2244)
  • 446885d feat(cluster): add node address map for TLS hostname remapping (#2155)
  • b2cde38 Replace matcher array macro with const generics (#2197)
  • 56bc260 Add logs to indicate when recovery future starts and completes (#2240)
  • f518e64 docs: Remove remarks about Redis docs to BF.MEXISTS being wrong (#2231)
  • Additional commits viewable in compare view

Dependabot compatibility score

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot show <dependency name> ignore conditions will show all of the ignore conditions of the specified dependency
  • @dependabot ignore this major version will close this PR and stop Dependabot creating any more for this major version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this minor version will close this PR and stop Dependabot creating any more for this minor version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this dependency will close this PR and stop Dependabot creating any more for this dependency (unless you reopen the PR or upgrade to it yourself)

Bumps [redis](https://github.com/redis-rs/redis-rs) from 1.2.1 to 1.5.0.
- [Release notes](https://github.com/redis-rs/redis-rs/releases)
- [Commits](redis-rs/redis-rs@redis-1.2.1...redis-1.5.0)

---
updated-dependencies:
- dependency-name: redis
  dependency-version: 1.5.0
  dependency-type: direct:production
  update-type: version-update:semver-minor
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot dependabot Bot added dependencies Pull requests that update a dependency file rust Pull requests that update rust code labels Aug 3, 2026
@verseghy-prow
verseghy-prow Bot requested a review from smrtrfszm August 3, 2026 04:25
@verseghy-prow

verseghy-prow Bot commented Aug 3, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: dependabot[bot]
Once this PR has been reviewed and has the lgtm label, please assign smrtrfszm for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@dependabot @github

dependabot Bot commented on behalf of github Aug 24, 2026

Copy link
Copy Markdown
Contributor Author

Superseded by #240.

@dependabot dependabot Bot closed this Aug 24, 2026
@dependabot
dependabot Bot deleted the dependabot/cargo/redis-1.5.0 branch August 24, 2026 04:30
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dependencies Pull requests that update a dependency file rust Pull requests that update rust code

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants