Skip to content

Update CODEOWNERS - #74

Merged
maddawik merged 1 commit into
mainfrom
dgvulncheck-patch-1
Aug 13, 2025
Merged

maddawik merged 1 commit into
mainfrom
dgvulncheck-patch-1

Conversation

@dgvulncheck

Copy link
Copy Markdown
Collaborator

add the sdk development team to the codeowners file

add the sdk development team to the codeowners file
@dgvulncheck
dgvulncheck requested a review from maddawik as a code owner August 13, 2025 15:59
@maddawik
maddawik merged commit 7785567 into main Aug 13, 2025
4 checks passed
@maddawik
maddawik deleted the dgvulncheck-patch-1 branch August 13, 2025 16:33
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants