Skip to content
This repository was archived by the owner on Aug 24, 2026. It is now read-only.

test do not merge - #147

Closed
mgoerens wants to merge 1 commit into
mainfrom
magotest
Closed

mgoerens wants to merge 1 commit into
mainfrom
magotest

Conversation

@mgoerens

Copy link
Copy Markdown
Collaborator

No description provided.

@exe-prow-github-app

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: mgoerens

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@mgoerens mgoerens closed this Sep 16, 2025
@mgoerens
mgoerens deleted the magotest branch March 5, 2026 14:24
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant