Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
✅ Pull request preview available for checkingBuilt without sensitive environment variables
To edit notification comments on pull requests, go to your Netlify project configuration. |
|
LGTM |
Okabe-Junya
left a comment
There was a problem hiding this comment.
/lgtm
/hold
(on behalf of cncf/glossary maintainer)
Let's unhold once cncf/glossary#3627 merged
|
LGTM label has been added. DetailsGit tree hash: 8da8d79e3393e620a66df840775d997c25a223e8 |
|
/assign reylejano |
|
The Kubernetes project currently lacks enough contributors to adequately respond to all PRs. This bot triages PRs according to the following rules:
You can:
Please send feedback to sig-contributor-experience at kubernetes/community. /lifecycle stale |
|
/remove-lifecycle stale |
|
/sig docs |
Thank you, Friendly reminder, still on it? |
Hey sure , I missed this notification , I have updated the PR now |
|
Closing this, I am not going to follow up on it. |
Summary
Validation
Fixes #55017