Skip to content

[Snyk] Security upgrade @docusaurus/preset-classic from 2.4.1 to 3.5.0 - #309

Open
hormes wants to merge 1 commit into
mainfrom
snyk-fix-146fc16c2ec6f33cd1373ec2b0aa0fe3
Open

hormes wants to merge 1 commit into
mainfrom
snyk-fix-146fc16c2ec6f33cd1373ec2b0aa0fe3

Conversation

@hormes

@hormes hormes commented Apr 13, 2026

Copy link
Copy Markdown
Member

snyk-top-banner

Snyk has created this PR to fix 3 vulnerabilities in the npm dependencies of this project.

Snyk changed the following file(s):

  • package.json
  • package-lock.json

Vulnerabilities that will be fixed with an upgrade:

Issue Score
high severity Arbitrary Code Injection
SNYK-JS-SERIALIZEJAVASCRIPT-570062
  706  
high severity Allocation of Resources Without Limits or Throttling
SNYK-JS-SERIALIZEJAVASCRIPT-15809196
  696  
high severity Regular Expression Denial of Service (ReDoS)
SNYK-JS-TRIM-1017038
  696  

Important

  • Check the changes in this PR to ensure they won't cause issues with your project.
  • Max score is 1000. Note that the real score may have changed since the PR was raised.
  • This PR was automatically created by Snyk using the credentials of a real user.

Note: You are seeing this because you or someone else with access to this repository has authorized Snyk to open fix PRs.

For more information:
🧐 View latest project report
📜 Customise PR templates
🛠 Adjust project settings
📚 Read about Snyk's upgrade logic


Learn how to fix vulnerabilities with free interactive lessons:

🦉 Allocation of Resources Without Limits or Throttling
🦉 Arbitrary Code Injection
🦉 Regular Expression Denial of Service (ReDoS)

@koordinator-bot
koordinator-bot Bot requested a review from ZiMengSheng April 13, 2026 10:57
@koordinator-bot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
To complete the pull request process, please ask for approval from hormes after the PR has been reviewed.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

This branch has not been deployed

No deployments
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants