Make a macOS user capable of being the service's user-account - #40
Draft
petemounce wants to merge 7 commits into
Draft
Make a macOS user capable of being the service's user-account#40petemounce wants to merge 7 commits into
petemounce wants to merge 7 commits into
Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
added 3 commits
September 3, 2020 10:22
petemounce
commented
Sep 4, 2020
| - name: make password for user | ||
| set_fact: | ||
| user_password: "{{ lookup('password', '/tmp/bk-agent-password length=32 chars=ascii_letters,digits,punctuation') }}" | ||
| buildkite_agent_user_password: "{{ lookup('password', '/tmp/bk-agent-password length=32 chars=ascii_letters,digits,punctuation') }}" |
Contributor
Author
There was a problem hiding this comment.
This was renamed to namespace the fact to the role.
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Changes
Now, running the role on macOS will create a user suitable for being the service's account.
That user has various sensible defaults set up
This is consistent with the behaviour
Verification
Local anka bake starting from a bare-OS image that does not have the
buildkite-agentcreated within it:verify that the buildkite-agent binary can start up and run when the user is created via this method. Screenshot of connected-to-bk agent page:
confirm all of the above via a physical macOS: