Skip to content

build(deps-dev): bump stylelint from 15.11.0 to 16.23.0 - #391

Closed
dependabot[bot] wants to merge 1 commit into
mainfrom
dependabot/npm_and_yarn/stylelint-16.23.0
Closed

dependabot[bot] wants to merge 1 commit into
mainfrom
dependabot/npm_and_yarn/stylelint-16.23.0

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Jul 30, 2025 •

Copy link
Copy Markdown
Contributor

Bumps stylelint from 15.11.0 to 16.23.0.

Release notes

Sourced from stylelint's releases.

16.23.0

It adds 5 new rules, 1 option to a rule and fixes 3 bugs. We've turned on the new rules in our standard config.

16.22.0

It adds 3 rule features and fixes 2 bugs.

16.21.1

It fixes 3 bugs.

16.21.0

It rounds out the message secondary option and marks it as stable, adds 1 new option to a rule, and fixes 4 bugs.

  • Added: non-experimental support for functional message secondary option (#8595) (@鈥媦biquitous).
  • Added: ignore: ["after-custom-property"] to custom-property-empty-line-before (#8627) (@鈥媔mkesin).
  • Fixed: cache configuration property not being respected (#8599) (@鈥媦biquitous).
  • Fixed: fastest-levenshtein import when bundling (#8621) (@鈥媜nigoetz).
  • Fixed: named-grid-areas-no-invalid false positives for interpolated grid-template-areas in CSS-in-JS (#8606) (@鈥媠hahana308).
  • Fixed: selector-pseudo-*-no-unknown false positives for the missing form control styling selectors (#8623) (@鈥媡yrann0us).

16.20.0

It fixes 2 bugs with the built-in rules.

16.19.1

It fixes 1 bug with the no-empty-source rule.

... (truncated)

Changelog

Sourced from stylelint's changelog.

16.23.0 - 2025-07-29

It adds 5 new rules, 1 option to a rule and fixes 3 bugs. We've turned on the new rules in our standard config.

16.22.0 - 2025-07-18

It adds 3 rule features and fixes 2 bugs.

16.21.1 - 2025-07-03

It fixes 3 bugs.

16.21.0 - 2025-06-19

It rounds out the message secondary option and marks it as stable, adds 1 new option to a rule, and fixes 4 bugs.

  • Added: non-experimental support for functional message secondary option (#8595) (@鈥媦biquitous).
  • Added: ignore: ["after-custom-property"] to custom-property-empty-line-before (#8627) (@鈥媔mkesin).
  • Fixed: cache configuration property not being respected (#8599) (@鈥媦biquitous).
  • Fixed: fastest-levenshtein import when bundling (#8621) (@鈥媜nigoetz).
  • Fixed: named-grid-areas-no-invalid false positives for interpolated grid-template-areas in CSS-in-JS (#8606) (@鈥媠hahana308).
  • Fixed: selector-pseudo-*-no-unknown false positives for the missing form control styling selectors (#8623) (@鈥媡yrann0us).

16.20.0 - 2025-05-29

It fixes 2 bugs with the built-in rules.

... (truncated)

Commits

Dependabot compatibility score

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot merge will merge this PR after your CI passes on it
  • @dependabot squash and merge will squash and merge this PR after your CI passes on it
  • @dependabot cancel merge will cancel a previously requested merge and block automerging
  • @dependabot reopen will reopen this PR if it is closed
  • @dependabot close will close this PR and stop Dependabot recreating it. You can achieve the same result by closing it manually
  • @dependabot show <dependency name> ignore conditions will show all of the ignore conditions of the specified dependency
  • @dependabot ignore this major version will close this PR and stop Dependabot creating any more for this major version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this minor version will close this PR and stop Dependabot creating any more for this minor version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this dependency will close this PR and stop Dependabot creating any more for this dependency (unless you reopen the PR or upgrade to it yourself)

@dependabot dependabot Bot added the dependencies Pull requests that update a dependency file label Jul 30, 2025
@vercel

vercel Bot commented Jul 30, 2025 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git 鈫楋笌

1 Skipped Deployment
Name Status Preview Comments Updated (UTC)
mdshare 猬滐笍 Ignored (Inspect) Visit Preview Aug 4, 2025 2:47am

@dependabot
dependabot Bot force-pushed the dependabot/npm_and_yarn/stylelint-16.23.0 branch from a1ce20d to c430263 Compare August 1, 2025 02:36
Bumps [stylelint](https://github.com/stylelint/stylelint) from 15.11.0 to 16.23.0.
- [Release notes](https://github.com/stylelint/stylelint/releases)
- [Changelog](https://github.com/stylelint/stylelint/blob/main/CHANGELOG.md)
- [Commits](stylelint/stylelint@15.11.0...16.23.0)

---
updated-dependencies:
- dependency-name: stylelint
  dependency-version: 16.23.0
  dependency-type: direct:development
  update-type: version-update:semver-major
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot
dependabot Bot force-pushed the dependabot/npm_and_yarn/stylelint-16.23.0 branch from c430263 to 089452d Compare August 4, 2025 02:47
@dependabot @github

dependabot Bot commented on behalf of github Aug 8, 2025

Copy link
Copy Markdown
Contributor Author

Superseded by #405.

@dependabot dependabot Bot closed this Aug 8, 2025
@dependabot
dependabot Bot deleted the dependabot/npm_and_yarn/stylelint-16.23.0 branch August 8, 2025 02:15
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dependencies Pull requests that update a dependency file

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants